Skip to content

[main] Update community files - #86

Merged
knative-prow[bot] merged 1 commit into
knative-extensions:mainfrom
knative-automation:auto-updates/update-community-main
Oct 9, 2026
Merged

knative-prow[bot] merged 1 commit into
knative-extensions:mainfrom
knative-automation:auto-updates/update-community-main

Conversation

@knative-automation

@knative-automation knative-automation commented May 21, 2026 •

Copy link
Copy Markdown
Contributor

new repo -aliok

Produced by: knative-extensions/knobots/actions/update-community

Details:


@knative-prow

knative-prow Bot commented May 21, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@knative-prow knative-prow Bot added do-not-merge/invalid-owners-file Indicates that a PR should not merge because it has an invalid OWNERS file in it. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels May 21, 2026

@gauron99 gauron99 left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@knative-prow knative-prow Bot added lgtm Indicates that a PR is ready to be merged. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels May 26, 2026
Signed-off-by: Knative Automation <automation@knative.team>
@knative-automation
knative-automation force-pushed the auto-updates/update-community-main branch from feff6a1 to c963c2e Compare October 9, 2026 09:34
@knative-prow knative-prow Bot removed the lgtm Indicates that a PR is ready to be merged. label Oct 9, 2026
@knative-prow

knative-prow Bot commented Oct 9, 2026

Copy link
Copy Markdown

New changes are detected. LGTM label has been removed.

@knative-prow

knative-prow Bot commented Oct 9, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@knative-prow knative-prow Bot removed the do-not-merge/invalid-owners-file Indicates that a PR should not merge because it has an invalid OWNERS file in it. label Oct 9, 2026
@knative-prow

knative-prow Bot commented Oct 9, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: gauron99, knative-automation

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow
knative-prow Bot merged commit 36700c5 into knative-extensions:main Oct 9, 2026
14 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. skip-review

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants